Skip to content

Add tenancy vw - #3

Open
mjudeikis wants to merge 1 commit into
mainfrom
tenancy.vw
Open

mjudeikis wants to merge 1 commit into
mainfrom
tenancy.vw

Conversation

@mjudeikis

Copy link
Copy Markdown
Contributor

POC for tenancy as virtual workspace (+ controller)

@kcp-ci-bot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants