Skip to content

feat: expose workspace name in list_kcp_workspaces - #1

Merged
mjudeikis merged 1 commit into
kcp-dev:mainfrom
cnvergence:workspace-name
Sep 10, 2026
Merged

mjudeikis merged 1 commit into
kcp-dev:mainfrom
cnvergence:workspace-name

Conversation

@cnvergence

Copy link
Copy Markdown
Member

No description provided.

Signed-off-by: Karol Szwaj <karol.szwaj@gmail.com>
@kcp-ci-bot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign cnvergence for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@mjudeikis
mjudeikis merged commit 40ad9c7 into kcp-dev:main Sep 10, 2026
9 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants