Skip to content

bump dependencies to @journeyapps scope - #29

Merged
dylanvorster merged 1 commit into
mainfrom
upgrade_dependencies
Aug 18, 2026
Merged

dylanvorster merged 1 commit into
mainfrom
upgrade_dependencies

Conversation

@dylanvorster

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented Aug 18, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 635ac04

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@journeyapps/common-sdk Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dylanvorster
dylanvorster merged commit b1e9c22 into main Aug 18, 2026
1 check passed
@dylanvorster
dylanvorster deleted the upgrade_dependencies branch August 18, 2026 22:42
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants