Skip to content

fix publishing - #25

Merged
dylanvorster merged 1 commit into
mainfrom
fix_publishing
Jul 28, 2026
Merged

fix publishing#25
dylanvorster merged 1 commit into
mainfrom
fix_publishing

Conversation

@dylanvorster

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented Jul 28, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 72b3575

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
@journeyapps-labs/common-docs Patch
@journeyapps-labs/common-ioc Patch
@journeyapps-labs/common-logger Patch
@journeyapps-labs/common-sdk Patch
@journeyapps-labs/common-tree Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dylanvorster
dylanvorster merged commit fdcc87c into main Jul 28, 2026
1 check passed
@dylanvorster
dylanvorster deleted the fix_publishing branch July 28, 2026 17:48
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants