Skip to content

Bump github.com/coreos/go-oidc/v3 from 3.20.0 to 3.21.0 - #265

Merged
gardener-prow[bot] merged 1 commit into
masterfrom
dependabot/go_modules/github.com/coreos/go-oidc/v3-3.21.0
Sep 8, 2026
Merged

Bump github.com/coreos/go-oidc/v3 from 3.20.0 to 3.21.0#265
gardener-prow[bot] merged 1 commit into
masterfrom
dependabot/go_modules/github.com/coreos/go-oidc/v3-3.21.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 7, 2026

Copy link
Copy Markdown
Contributor

Bumps github.com/coreos/go-oidc/v3 from 3.20.0 to 3.21.0.

Release notes

Sourced from github.com/coreos/go-oidc/v3's releases.

v3.21.0

What's Changed

New Contributors

Full Changelog: coreos/go-oidc@v3.20.0...v3.21.0

Commits
  • c914bd3 oidc: ignore JWKs with unsupported key types rather than failing
  • See full diff in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [github.com/coreos/go-oidc/v3](https://github.com/coreos/go-oidc) from 3.20.0 to 3.21.0.
- [Release notes](https://github.com/coreos/go-oidc/releases)
- [Commits](coreos/go-oidc@v3.20.0...v3.21.0)

---
updated-dependencies:
- dependency-name: github.com/coreos/go-oidc/v3
  dependency-version: 3.21.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added the kind/enhancement Enhancement, improvement, extension label Sep 7, 2026
@dependabot
dependabot Bot requested a review from a team as a code owner September 7, 2026 19:23
@dependabot dependabot Bot added the kind/enhancement Enhancement, improvement, extension label Sep 7, 2026
@gardener-prow gardener-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 7, 2026

@theoddora theoddora left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@gardener-prow gardener-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 8, 2026
@gardener-prow

gardener-prow Bot commented Sep 8, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 6460c4698e904bb88d9c15178949d20d1a4a564d

@gardener-prow

gardener-prow Bot commented Sep 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: theoddora

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 8, 2026
@gardener-prow
gardener-prow Bot merged commit 108a6a1 into master Sep 8, 2026
19 checks passed
@dependabot
dependabot Bot deleted the dependabot/go_modules/github.com/coreos/go-oidc/v3-3.21.0 branch September 8, 2026 06:21
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/enhancement Enhancement, improvement, extension lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant