Skip to content

chore(deps): update konflux references - #350

Merged
michael-valdron merged 1 commit into
mainfrom
konflux/references/main
Apr 24, 2026
Merged

michael-valdron merged 1 commit into
mainfrom
konflux/references/main

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Apr 16, 2026 •

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change
quay.io/konflux-ci/tekton-catalog/task-apply-tags (source, changelog) aa62b41 → a291081
quay.io/konflux-ci/tekton-catalog/task-build-image-index (source, changelog) ae3fa44 → 550afde
quay.io/konflux-ci/tekton-catalog/task-buildah-oci-ta (source, changelog) 9741037 → 681d9f6
quay.io/konflux-ci/tekton-catalog/task-clair-scan (source, changelog) 8992475 → cd49cde
quay.io/konflux-ci/tekton-catalog/task-clamav-scan (source, changelog) 4b0f83c → 171eca5
quay.io/konflux-ci/tekton-catalog/task-ecosystem-cert-preflight-checks (source, changelog) 2d439dc → 2468c01
quay.io/konflux-ci/tekton-catalog/task-git-clone-oci-ta (source, changelog) 2c388d2 → 13d49df
quay.io/konflux-ci/tekton-catalog/task-init (source, changelog) 288f310 → b797dd4
quay.io/konflux-ci/tekton-catalog/task-prefetch-dependencies-oci-ta (source, changelog) 9917d11 → 5aa530b
quay.io/konflux-ci/tekton-catalog/task-push-dockerfile-oci-ta (source, changelog) 1bc2d0f → 7855471
quay.io/konflux-ci/tekton-catalog/task-rpms-signature-scan (source, changelog) 89c2bfe → ce4bace
quay.io/konflux-ci/tekton-catalog/task-sast-snyk-check-oci-ta (source, changelog) 2ad986f → 8f3ecbe
quay.io/konflux-ci/tekton-catalog/task-show-sbom (source, changelog) 04994df → a7346ed
quay.io/konflux-ci/tekton-catalog/task-source-build-oci-ta (source, changelog) 362f047 → 0917cfc

Configuration

📅 Schedule: Branch creation - At 07:00 PM through 11:59 PM and 12:00 AM through 02:59 AM, only on Tuesday and Thursday ( * 19-23,0-2 * * 2,4 ) in timezone America/Toronto, Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@openshift-ci
openshift-ci Bot requested review from Jdubrick and thepetk April 16, 2026 05:20
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/references/main branch 4 times, most recently from c0a3eaf to 8a43ef1 Compare April 23, 2026 05:18
Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/references/main branch from 8a43ef1 to 00d0a56 Compare April 24, 2026 01:18
@openshift-ci

openshift-ci Bot commented Apr 24, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: michael-valdron, red-hat-konflux[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@michael-valdron
michael-valdron merged commit e37752f into main Apr 24, 2026
11 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant