Skip to content

chore(deps): update makefile modules to 2a76aa9 - #2316

Merged
cert-manager-prow[bot] merged 1 commit into
masterfrom
renovate/makefile-modules
Oct 5, 2026
Merged

cert-manager-prow[bot] merged 1 commit into
masterfrom
renovate/makefile-modules

Conversation

@renovate

@renovate renovate Bot commented Oct 5, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Update Change
boilerplate digest f2fb9c3 → 2a76aa9
generate-verify digest f2fb9c3 → 2a76aa9
help digest f2fb9c3 → 2a76aa9
klone digest f2fb9c3 → 2a76aa9
repository-base digest f2fb9c3 → 2a76aa9
tools digest f2fb9c3 → 2a76aa9

Configuration

📅 Schedule: (in timezone Europe/London)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
@renovate renovate Bot added dependencies Pull requests that update a dependency file ok-to-test skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. labels Oct 5, 2026
@cert-manager-prow cert-manager-prow Bot added the dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. label Oct 5, 2026
@cert-manager-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign erikgb for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@cert-manager-prow cert-manager-prow Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Oct 5, 2026
@netlify

netlify Bot commented Oct 5, 2026

Copy link
Copy Markdown

✅ Deploy Preview for cert-manager ready!

Name Link
🔨 Latest commit c61c9ff
🔍 Latest deploy log https://app.netlify.com/projects/cert-manager/deploys/6ac41bb89300d80008fb2d5b
😎 Deploy Preview https://deploy-preview-2316--cert-manager.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@cert-manager-prow
cert-manager-prow Bot merged commit c12e20b into master Oct 5, 2026
7 checks passed
@renovate
renovate Bot deleted the renovate/makefile-modules branch October 5, 2026 21:56
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. dependencies Pull requests that update a dependency file ok-to-test size/M Denotes a PR that changes 30-99 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants