Skip to content

[CI] Bump base image to latest available version on main - #772

Merged
cert-manager-prow[bot] merged 1 commit into
mainfrom
base-image-bump-main
Oct 7, 2026
Merged

cert-manager-prow[bot] merged 1 commit into
mainfrom
base-image-bump-main

Conversation

@github-actions

@github-actions github-actions Bot commented Oct 7, 2026

Copy link
Copy Markdown
Contributor

This PR is auto-generated to bump the base images used for our OCI builds.

Signed-off-by: cert-manager-bot <cert-manager-bot@users.noreply.github.com>
@github-actions github-actions Bot added the skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. label Oct 7, 2026
@cert-manager-prow cert-manager-prow Bot added the dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. label Oct 7, 2026
@cert-manager-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign jakexks for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@cert-manager-prow cert-manager-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Oct 7, 2026
@cert-manager-prow
cert-manager-prow Bot merged commit 82d2ae9 into main Oct 7, 2026
2 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant