Skip to content

chore(deps): update makefile modules to c153741 - #174

Merged
cert-manager-prow[bot] merged 1 commit into
mainfrom
renovate/makefile-modules
Oct 3, 2026
Merged

cert-manager-prow[bot] merged 1 commit into
mainfrom
renovate/makefile-modules

Conversation

@renovate

@renovate renovate Bot commented Oct 3, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Update Change
boilerplate digest 9096105 → c153741
generate-verify digest 9096105 → c153741
go digest 9096105 → c153741
help digest 9096105 → c153741
klone digest 9096105 → c153741
repository-base digest 9096105 → c153741
tools digest 9096105 → c153741

Configuration

📅 Schedule: (in timezone Europe/London)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
@renovate renovate Bot added dependencies ok-to-test skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. labels Oct 3, 2026
@cert-manager-prow cert-manager-prow Bot added the dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. label Oct 3, 2026
@cert-manager-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign inteon for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@cert-manager-prow cert-manager-prow Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Oct 3, 2026
@cert-manager-prow
cert-manager-prow Bot merged commit c818edf into main Oct 3, 2026
6 checks passed
@renovate
renovate Bot deleted the renovate/makefile-modules branch October 3, 2026 16:39
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. dependencies ok-to-test size/M Denotes a PR that changes 30-99 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants