Skip to content

Add O957 To Code Owners File - #263

Open
O957 wants to merge 1 commit into
mainfrom
upx3_add_self_to_code_owners
Open

Add O957 To Code Owners File#263
O957 wants to merge 1 commit into
mainfrom
upx3_add_self_to_code_owners

Conversation

@O957

@O957 O957 commented Sep 4, 2026

Copy link
Copy Markdown
Contributor

Done as suggested by DHM, given recent contributions.

@dylanhmorris

Copy link
Copy Markdown
Collaborator

Not just CODEOWNERS. You should add yourself as a contributor in the DESCRIPTION

@codecov

codecov Bot commented Sep 4, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 75.51%. Comparing base (56ac56c) to head (1572db7).

Additional details and impacted files
@@           Coverage Diff           @@
##             main     #263   +/-   ##
=======================================
  Coverage   75.51%   75.51%           
=======================================
  Files          37       37           
  Lines        2091     2091           
=======================================
  Hits         1579     1579           
  Misses        512      512           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants